Skip to content

[pre-commit.ci] pre-commit autoupdate - #1939

Open
pre-commit-ci[bot] wants to merge 1 commit into
masterfrom
pre-commit-ci-update-config
Open

[pre-commit.ci] pre-commit autoupdate#1939
pre-commit-ci[bot] wants to merge 1 commit into
masterfrom
pre-commit-ci-update-config

Conversation

@pre-commit-ci

@pre-commit-ci pre-commit-ci Bot commented Sep 7, 2026

Copy link
Copy Markdown
Contributor

updates:
- [github.com/pre-commit/mirrors-clang-format: v22.1.8 → v23.1.0](pre-commit/mirrors-clang-format@v22.1.8...v23.1.0)
@deepin-ci-robot

Copy link
Copy Markdown
Collaborator

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: pre-commit-ci[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@deepin-ci-robot

Copy link
Copy Markdown
Collaborator

Hi @pre-commit-ci[bot]. Thanks for your PR.

I'm waiting for a OpenAtom-Linyaps member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant