Skip to content

Fix corrupted CODEOWNERS - #57

Merged
cygnusb merged 1 commit into
mainfrom
fix-codeowners-garbage
Aug 14, 2026
Merged

cygnusb merged 1 commit into
mainfrom
fix-codeowners-garbage

Conversation

@cygnusb

@cygnusb cygnusb commented Aug 14, 2026

Copy link
Copy Markdown
Owner

Removes stray garbage bytes accidentally written before the CODEOWNERS entry.

@cygnusb
cygnusb merged commit 71d594c into main Aug 14, 2026
@cygnusb
cygnusb deleted the fix-codeowners-garbage branch August 14, 2026 06:39
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant